Skip to content

openterface-qt: init at 0.0.6 - #365464

Merged
lukegb merged 1 commit into
NixOS:masterfrom
wlcx:init-openterface
Jan 19, 2025
Merged

lukegb merged 1 commit into
NixOS:masterfrom
wlcx:init-openterface

Conversation

@wlcx

@wlcx wlcx commented Dec 16, 2024

Copy link
Copy Markdown
Contributor

Add the openterface mini kvm host application

Things done

  • Built on platform(s)
    • x86_64-linux
    • aarch64-linux
    • x86_64-darwin
    • aarch64-darwin
  • For non-Linux: Is sandboxing enabled in nix.conf? (See Nix manual)
    • sandbox = relaxed
    • sandbox = true
  • Tested, as applicable:
  • Tested compilation of all packages that depend on this change using nix-shell -p nixpkgs-review --run "nixpkgs-review rev HEAD". Note: all changes have to be committed, also see nixpkgs-review usage
  • Tested basic functionality of all binary files (usually in ./result/bin/)
  • 25.05 Release Notes (or backporting 24.11 and 25.05 Release notes)
    • (Package updates) Added a release notes entry if the change is major or breaking
    • (Module updates) Added a release notes entry if the change is significant
    • (Module addition) Added a release notes entry if adding a new NixOS module
  • Fits CONTRIBUTING.md.

Add a 👍 reaction to pull requests you find important.

@github-actions github-actions Bot added 10.rebuild-darwin: 0 This PR does not cause any packages to rebuild on Darwin. 10.rebuild-linux: 1-10 This PR causes between 1 and 10 packages to rebuild on Linux. labels Dec 16, 2024
@ofborg ofborg Bot added 8.has: package (new) This PR adds a new package 11.by: package-maintainer This PR was created by a maintainer of all the package it changes. 10.rebuild-linux: 1 This PR causes 1 package to rebuild on Linux. labels Dec 16, 2024
@wlcx
wlcx requested a review from lukegb December 17, 2024 14:41
Comment thread pkgs/by-name/op/openterface-qt/package.nix Outdated
Comment thread pkgs/by-name/op/openterface-qt/package.nix Outdated
@wlcx wlcx changed the title openterface-qt: init at 0.0.5 openterface-qt: init at 0.0.6 Jan 19, 2025
@wlcx
wlcx force-pushed the init-openterface branch from 4758c8b to 5d580a3 Compare January 19, 2025 00:34
@wlcx

wlcx commented Jan 19, 2025

Copy link
Copy Markdown
Contributor Author

Thanks for the review @lukegb - made updates. I also switched to the finalattrs style of avoiding duplication :)

@github-actions github-actions Bot removed the 10.rebuild-linux: 1 This PR causes 1 package to rebuild on Linux. label Jan 19, 2025
@lukegb
lukegb merged commit 15d6e69 into NixOS:master Jan 19, 2025
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

8.has: package (new) This PR adds a new package 10.rebuild-darwin: 0 This PR does not cause any packages to rebuild on Darwin. 10.rebuild-linux: 1-10 This PR causes between 1 and 10 packages to rebuild on Linux. 11.by: package-maintainer This PR was created by a maintainer of all the package it changes.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants