Skip to content

feat: notion piece bounty - #8217

Merged
kishanprmr merged 12 commits into
activepieces:mainfrom
ezhil56x:notion-mcp
Jul 8, 2025
Merged

feat: notion piece bounty#8217
kishanprmr merged 12 commits into
activepieces:mainfrom
ezhil56x:notion-mcp

Conversation

@ezhil56x

@ezhil56x ezhil56x commented Jul 1, 2025

Copy link
Copy Markdown
Contributor

What does this PR do?

Added few some actions and triggers to Notion Pieces

  • New Comment
  • Updated Page
  • Archive Database Item
  • Restore Database Item
  • Add Comment
  • Retrieve Database
  • Get Page Comments
  • Find Page
Notion.mp4

Fixes #8211
/claim #8211

@algora-pbc algora-pbc Bot mentioned this pull request Jul 1, 2025
@CLAassistant

CLAassistant commented Jul 1, 2025

Copy link
Copy Markdown

CLA assistant check
All committers have signed the CLA.

Comment thread packages/pieces/community/notion/src/lib/actions/add-comment.ts Outdated
Comment thread packages/pieces/community/notion/src/lib/actions/restore-database-item.ts Outdated
Comment thread packages/pieces/community/notion/src/lib/actions/retrieve-database.ts Outdated

@onyedikachi-david onyedikachi-david left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Thanks for the PR @ezhil56x, please note these:

Comment thread packages/pieces/community/notion/src/lib/triggers/new-comment.ts Outdated
Comment thread packages/pieces/community/notion/src/lib/triggers/new-comment.ts Outdated
@ezhil56x

ezhil56x commented Jul 4, 2025

Copy link
Copy Markdown
Contributor Author

I'll make the requested changes at the earliest, thanks!

@ezhil56x

ezhil56x commented Jul 6, 2025

Copy link
Copy Markdown
Contributor Author

@onyedikachi-david I have made the required changes.
For restore-database-item.ts, notion doesn't give option to include rows that are archived in notion.search

@nx-cloud

nx-cloud Bot commented Jul 8, 2025

Copy link
Copy Markdown

View your CI Pipeline Execution ↗ for commit 8281738

Command Status Duration Result
nx run-many --target=build -c production --proj... ✅ Succeeded 10s View ↗
nx affected --target=build -c production --excl... ✅ Succeeded <1s View ↗
nx run-many --target=lint --projects=pieces-not... ✅ Succeeded 1m View ↗
nx affected --target=lint --exclude=pieces-* --... ✅ Succeeded <1s View ↗

☁️ Nx Cloud last updated this comment at 2025-07-08 13:40:08 UTC

@kishanprmr kishanprmr changed the title feat: notion piece feat: notion piece bounty Jul 8, 2025
@kishanprmr

Copy link
Copy Markdown
Collaborator

@allcontributors please add @ezhil56x for plugin

@allcontributors

Copy link
Copy Markdown
Contributor

@kishanprmr

I've put up a pull request to add @ezhil56x! 🎉

@kishanprmr
kishanprmr merged commit 8cdf950 into activepieces:main Jul 8, 2025
@github-project-automation github-project-automation Bot moved this from In Progress to Done in Sprint (6th July - 13th July) Jul 8, 2025
@ezhil56x
ezhil56x deleted the notion-mcp branch July 9, 2025 11:31
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

No open projects

Development

Successfully merging this pull request may close these issues.

[MCP] Notion

4 participants