Skip to content

[Bigquery] More retries is better retries - #62041

Merged
Francis Genet (frifriSF59) merged 5 commits into
masterfrom
frifri/retry_503_bq
Jun 24, 2025
Merged

[Bigquery] More retries is better retries#62041
Francis Genet (frifriSF59) merged 5 commits into
masterfrom
frifri/retry_503_bq

Conversation

@frifriSF59

@frifriSF59 Francis Genet (frifriSF59) commented Jun 24, 2025

Copy link
Copy Markdown
Contributor

What

More retries to try and not run into the "Visibility check was unavailable" error. maybe?

Important

Auto-merge enabled.

This PR is set to merge automatically when all requirements are met.

@vercel

vercel Bot commented Jun 24, 2025

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for Git ↗︎

1 Skipped Deployment
Name Status Preview Comments Updated (UTC)
airbyte-docs ⬜️ Ignored (Inspect) Visit Preview Jun 24, 2025 6:31pm

@github-actions

Copy link
Copy Markdown
Contributor

👋 Greetings, Contributor!

Here are some helpful tips and reminders for your convenience.

Helpful Resources

PR Slash Commands

Airbyte Maintainers (that's you!) can execute the following slash commands on your PR:

  • /format-fix - Fixes most formatting issues.
  • /bump-version - Bumps connector versions.
    • You can specify a custom changelog by passing changelog. Example: /bump-version changelog="My cool update"
    • Leaving the changelog arg blank will auto-populate the changelog from the PR title.
  • /run-cat-tests - Runs legacy CAT tests (Connector Acceptance Tests)
  • /build-connector-images - Builds and publishes a pre-release docker image for the modified connector(s).

📝 Edit this welcome message.

@github-actions

Copy link
Copy Markdown
Contributor

destination-bigquery Connector Test Results

226 tests   162 ✅  25m 10s ⏱️
 11 suites   64 💤
 11 files      0 ❌

Results for commit 396c2d8.

@frifriSF59
Francis Genet (frifriSF59) enabled auto-merge (squash) June 24, 2025 18:43
.setInitialRpcTimeoutDuration(java.time.Duration.ofMillis(50_000L))
.setRpcTimeoutMultiplier(1.0)
.setMaxRpcTimeoutDuration(java.time.Duration.ofMillis(50_000L))
.setMaxAttempts(15)

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

nit: add a comment that most of these are the defaults, and this is the only difference

@frifriSF59
Francis Genet (frifriSF59) merged commit 8369d85 into master Jun 24, 2025
@frifriSF59
Francis Genet (frifriSF59) deleted the frifri/retry_503_bq branch June 24, 2025 18:44
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants