Skip to content

Avoid rebuilds on makefile change - #421

Merged
c-w merged 3 commits into
masterfrom
build
Apr 29, 2020
Merged

Avoid rebuilds on makefile change#421
c-w merged 3 commits into
masterfrom
build

Conversation

@c-w

@c-w c-w commented Apr 29, 2020

Copy link
Copy Markdown
Member

Resolves #419

@codecov-io

Copy link
Copy Markdown

Codecov Report

Merging #421 into master will not change coverage.
The diff coverage is n/a.

Impacted file tree graph

@@           Coverage Diff           @@
##           master     #421   +/-   ##
=======================================
  Coverage   76.52%   76.52%           
=======================================
  Files          43       43           
  Lines        2718     2718           
=======================================
  Hits         2080     2080           
  Misses        638      638           

Continue to review full report at Codecov.

Legend - Click here to learn more
Δ = absolute <relative> (impact), ø = not affected, ? = missing data
Powered by Codecov. Last update f85a763...c2d0c7d. Read the comment docs.

@c-w
c-w merged commit 1df9dad into master Apr 29, 2020
@c-w
c-w deleted the build branch April 29, 2020 04:45
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

Decouple app build and linters

2 participants