support promote query to service - #1506
Merged
Merged
Conversation
🦋 Changeset detectedLatest commit: 9cc3f10 The changes in this PR will be included in the next version bump. This PR includes changesets to release 34 packages
Not sure what this means? Click here to learn what changesets are. Click here if you're a maintainer who wants to add another changeset to this PR |
25 tasks
Codecov Report
@@ Coverage Diff @@
## master #1506 +/- ##
==========================================
- Coverage 40.61% 40.35% -0.26%
==========================================
Files 1248 1252 +4
Lines 55588 56074 +486
Branches 12566 12687 +121
==========================================
+ Hits 22579 22631 +52
- Misses 32926 33361 +435
+ Partials 83 82 -1
|
akphi
marked this pull request as ready for review
September 29, 2022 21:19
akphi
requested review from
MauricioUyaguari and
pierredebelen
as code owners
September 29, 2022 21:19
MauricioUyaguari
approved these changes
Sep 29, 2022
3 tasks
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
Summary
Legend Querysetup page2charactersProjectfromrepotoopen bookProductionize an existing query(with iconFaArrowAltCircleUp), which let users pick a query (their query only), and then, based on that to go to a route inLegend Studio-/extensions/productionize-query/:${queryId}Studio-/productionize-query/:${queryId}delete the query upon completion if they are the owner of the query, after this, show a proceed to the mode where they can view their service in focus-mode or in the workspace, and prompt them to fix their workspace (if there was compilation issues detected) and then proceed with creating a review
EditExistingQuerymode, if the project cannot be resolved fromprojectId prefix, pick the default instanceofStudioEditExistingQuery, only owner can promote to SDLC flow -> Disallow promote to service if not ownerownerof query promotion nowupdate-service-querymode to tell users what to expect: i.e.Hey, your service has been cloned, .. but it's not committed, please have it reviewed and merged, etc. As of now, you can keep working on it, using the following link, have a button to go to that service, or abortLast but not least, thanks @MauricioUyaguari a lot for the ground work in #1375, covered a lot of cases I could have missed. I simplified the hell out of it though 😢
How did you test this change?
Productionize query (happy path)
productionize.query.mov
Productionize query with compilation issue
productionize.query.with.compilation.issue.mov
New
Projecticon - Project is an Open Book!