Skip to content
This repository was archived by the owner on Aug 30, 2026. It is now read-only.

feat: Create CODEOWNERS - #27

Merged
crwilcox merged 2 commits into
masterfrom
czahedi-patch-1
May 11, 2020
Merged

feat: Create CODEOWNERS#27
crwilcox merged 2 commits into
masterfrom
czahedi-patch-1

Conversation

@czahedi

@czahedi czahedi commented May 6, 2020

Copy link
Copy Markdown
Contributor

Adding owner team

Adding owner team
@googlebot googlebot added the cla: yes This human has signed the Contributor License Agreement. label May 6, 2020
@czahedi czahedi changed the title Create CODEOWNERS feat: Create CODEOWNERS May 11, 2020
@czahedi
czahedi requested a review from crwilcox May 11, 2020 21:36
@crwilcox
crwilcox merged commit 2b63746 into master May 11, 2020
@tseaver
tseaver deleted the czahedi-patch-1 branch August 5, 2020 20:46
Sign up for free to subscribe to this conversation on GitHub. Already have an account? Sign in.

Labels

cla: yes This human has signed the Contributor License Agreement.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants