feat: support setting rendering templates for related posts on category - #6106
Merged
Conversation
Member
Author
|
Hi @ruibaby ,需要为分类增加文章模板的选择项,帮忙提供一下 UI |
Codecov ReportAttention: Patch coverage is
Additional details and impacted files@@ Coverage Diff @@
## main #6106 +/- ##
============================================
- Coverage 56.91% 56.45% -0.47%
- Complexity 3319 3551 +232
============================================
Files 587 625 +38
Lines 18968 20988 +2020
Branches 1401 1484 +83
============================================
+ Hits 10795 11848 +1053
- Misses 7594 8537 +943
- Partials 579 603 +24 ☔ View full report in Codecov by Sentry. |
ruibaby
marked this pull request as ready for review
June 20, 2024 07:03
Member
Done |
|
[APPROVALNOTIFIER] This PR is APPROVED This pull-request has been approved by: JohnNiang, ruibaby The full list of commands accepted by this bot can be found here. The pull request process is described here DetailsNeeds approval from an approver in each of these files:
Approvers can indicate their approval by writing |
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
What type of PR is this?
/kind feature
/area core
/milestone 2.17.x
What this PR does / why we need it:
支持在分类上为关联的文章统一设置渲染模板
现在文章的模板生效顺序为:
Which issue(s) this PR fixes:
Fixes #6101
Does this PR introduce a user-facing change?