feat: add plugin detail modal as global component - #6233
Merged
f2c-ci-robot[bot] merged 3 commits intoJul 1, 2024
Conversation
Signed-off-by: Ryan Wang <i@ryanc.cc>
Signed-off-by: Ryan Wang <i@ryanc.cc>
Codecov ReportAll modified and coverable lines are covered by tests ✅
Additional details and impacted files@@ Coverage Diff @@
## main #6233 +/- ##
============================================
+ Coverage 54.51% 57.18% +2.67%
- Complexity 3523 3720 +197
============================================
Files 646 648 +2
Lines 21862 21979 +117
Branches 1528 1538 +10
============================================
+ Hits 11917 12569 +652
+ Misses 9328 8801 -527
+ Partials 617 609 -8 ☔ View full report in Codecov by Sentry. |
|
[APPROVALNOTIFIER] This PR is APPROVED This pull-request has been approved by: guqing, JohnNiang The full list of commands accepted by this bot can be found here. The pull request process is described here DetailsNeeds approval from an approver in each of these files:
Approvers can indicate their approval by writing |
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.

What type of PR is this?
/area ui
/kind feature
/milestone 2.17.x
What this PR does / why we need it:
添加 PluginDetailModal,用于打开插件的设置界面。并在扩展设置页面适配以测试。
Which issue(s) this PR fixes:
Fixes #6232
Does this PR introduce a user-facing change?