implementation of closed loop flow control - #6077
Merged
Merged
Conversation
kubeedge-bot
requested review from
WillardHu,
daixiang0,
fisherxu,
kadisi and
kevin-wangzefeng
January 13, 2025 03:03
tangming1996
force-pushed
the
feature/close-loop
branch
from
January 13, 2025 03:08
a1b847c to
6e40a02
Compare
Collaborator
|
Is it possible to submit another PR for the proposal? |
Shelley-BaoYue
left a comment
Collaborator
There was a problem hiding this comment.
We can put the proposal as another separate PR.
tangming1996
force-pushed
the
feature/close-loop
branch
from
January 20, 2025 06:21
6e40a02 to
6199d63
Compare
Contributor
Author
|
The proposal is separated into a separate pr #6097 |
Signed-off-by: ming.tang <ming.tang@daocloud.io>
tangming1996
force-pushed
the
feature/close-loop
branch
from
March 24, 2025 02:44
6199d63 to
3680c7c
Compare
Collaborator
|
/lgtm |
Collaborator
|
/lgtm |
Collaborator
|
/approve |
Collaborator
|
[APPROVALNOTIFIER] This PR is APPROVED This pull-request has been approved by: WillardHu The full list of commands accepted by this bot can be found here. The pull request process is described here DetailsNeeds approval from an approver in each of these files:
Approvers can indicate their approval by writing |
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
What type of PR is this?
/kind feature
What this PR does / why we need it:
adds closed loop flow control to have zone based routing
Which issue(s) this PR fixes:
Fixes #
#5755
Special notes for your reviewer:
Does this PR introduce a user-facing change?: