bump containerd version to v1.5.0 - #606
k8s-ci-robot merged 1 commit into
Conversation
Signed-off-by: Priyanka Saggu <priyankasaggu11929@gmail.com>
|
Welcome @Priyankasaggu11929! |
|
Hi @Priyankasaggu11929. Thanks for your PR. I'm waiting for a kubernetes-sigs member to verify that this patch is reasonable to test. If it is, they should reply with Once the patch is verified, the new status will be reflected by the I understand the commands that are listed here. DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. |
|
cc: @dims |
|
/ok-to-test |
|
@dims: GitHub didn't allow me to assign the following users: SanikaGawhane. Note that only kubernetes-sigs members, repo collaborators and people who have commented on this issue/PR can be assigned. Additionally, issues/PRs can only have 10 assignees at the same time. DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. |
|
/lgtm |
|
/approve |
|
[APPROVALNOTIFIER] This PR is APPROVED This pull-request has been approved by: codenrhoden, Priyankasaggu11929 The full list of commands accepted by this bot can be found here. The pull request process is described here DetailsNeeds approval from an approver in each of these files:
Approvers can indicate their approval by writing |
What this PR does / why we need it:
This PR bumps the
containerdversion to current GA version, v1.5.0Additional context
Link to the new release: https://github.com/containerd/containerd/releases/tag/v1.5.0
Signed-off-by: Priyanka Saggu priyankasaggu11929@gmail.com