Skip to content

Makefile target to build and push image for release - #138

Merged
k8s-ci-robot merged 1 commit into
kubernetes:masterfrom
nckturner:build-release
Nov 13, 2020
Merged

k8s-ci-robot merged 1 commit into
kubernetes:masterfrom
nckturner:build-release

Conversation

@nckturner

Copy link
Copy Markdown
Contributor

What type of PR is this?

Uncomment only one, leave it on its own line:

/kind api-change
/kind bug
/kind cleanup
/kind design
/kind documentation
/kind failing-test

/kind feature

/kind flake

What this PR does / why we need it:

We want to build and push containers using prow post-submits. This is a first pass at a script that takes care of tag format validation and passing GCR credentials to docker. Simplified version of the vsphere cloud provider release.sh script and makefile.

Which issue(s) this PR fixes:

Fixes #135

Special notes for your reviewer:

Does this PR introduce a user-facing change?:

NONE

@k8s-ci-robot k8s-ci-robot added do-not-merge/work-in-progress Indicates that a PR should not merge because it is a work in progress. release-note-none Denotes a PR that doesn't merit a release note. kind/feature Categorizes issue or PR as related to a new feature. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. size/L Denotes a PR that changes 100-499 lines, ignoring generated files. labels Sep 30, 2020
@k8s-ci-robot k8s-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Sep 30, 2020
Comment thread Makefile Outdated
# See the License for the specific language governing permissions and
# limitations under the License.
#
.EXPORT_ALL_VARIABLES:

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

TIL

Copy link
Copy Markdown
Contributor Author

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

handy, isn't it?

@k8s-ci-robot k8s-ci-robot added size/M Denotes a PR that changes 30-99 lines, ignoring generated files. and removed size/L Denotes a PR that changes 100-499 lines, ignoring generated files. labels Oct 16, 2020
@randomvariable

Copy link
Copy Markdown
Member

Looking good. I think we can iterate once we have cloudbuild enabled in test-infra?

@nckturner

Copy link
Copy Markdown
Contributor Author

Looking good. I think we can iterate once we have cloudbuild enabled in test-infra?

Yup.

@nckturner nckturner changed the title WIP Makefile target to build and push image for release Makefile target to build and push image for release Nov 13, 2020
@k8s-ci-robot k8s-ci-robot removed the do-not-merge/work-in-progress Indicates that a PR should not merge because it is a work in progress. label Nov 13, 2020

@M00nF1sh M00nF1sh left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/lgtm

@k8s-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: M00nF1sh, nckturner

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@k8s-ci-robot k8s-ci-robot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Nov 13, 2020
@k8s-ci-robot
k8s-ci-robot merged commit 1be8de7 into kubernetes:master Nov 13, 2020
@nckturner
nckturner deleted the build-release branch November 13, 2020 06:49
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. kind/feature Categorizes issue or PR as related to a new feature. lgtm "Looks good to me", indicates that a PR is ready to be merged. release-note-none Denotes a PR that doesn't merit a release note. size/M Denotes a PR that changes 30-99 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

Compile aws-cloud-controller-manager during prow CI

4 participants