Skip to content

Add script to generate changelog - #179

Merged
k8s-ci-robot merged 1 commit into
kubernetes:masterfrom
ayberk:changelog_script
Feb 22, 2021
Merged

Add script to generate changelog#179
k8s-ci-robot merged 1 commit into
kubernetes:masterfrom
ayberk:changelog_script

Conversation

@ayberk

@ayberk ayberk commented Feb 19, 2021

Copy link
Copy Markdown
Contributor

What type of PR is this?
/kind documentation

What this PR does / why we need it:
Adds a script to generate changelog between two tags. Example output between v1.19 tag and HEAD:

➜  cloud-provider-aws git:(changelog_script) ✗ 02/19 14:14 python3 hack/changelog.py --token $TOKEN --print-only --section-title Release --range v1.19.0-alpha.1.. 
Release
* Replace book homepage with README ([#176](https://github.com/kubernetes/cloud-provider-aws/pull/176), @ayberk)
* update klog library from 2.4.0 to 2.5.0 ([#170](https://github.com/kubernetes/cloud-provider-aws/pull/170), @dineshkumar181094)
* Add documentation for cred provider ([#174](https://github.com/kubernetes/cloud-provider-aws/pull/174), @ayberk)
* Add released versions to README ([#175](https://github.com/kubernetes/cloud-provider-aws/pull/175), @nckturner)
* feat: Helm chart for aws cloud controller manager ([#173](https://github.com/kubernetes/cloud-provider-aws/pull/173), @JESWINKNINAN)
* Merge legacy provider ([#160](https://github.com/kubernetes/cloud-provider-aws/pull/160), @nckturner)
* Add wongma7 to owners ([#172](https://github.com/kubernetes/cloud-provider-aws/pull/172), @nckturner)
* tags: initial implementation of tags ([#149](https://github.com/kubernetes/cloud-provider-aws/pull/149), @nicolehanjing)
* Migrate to mkdocs ([#167](https://github.com/kubernetes/cloud-provider-aws/pull/167), @ayberk)
* Update Documentation ([#165](https://github.com/kubernetes/cloud-provider-aws/pull/165), @ayberk)
* Add ECR creds provider ([#157](https://github.com/kubernetes/cloud-provider-aws/pull/157), @ayberk)
* Pass in cloud config file to initialize cloud provider ([#164](https://github.com/kubernetes/cloud-provider-aws/pull/164), @nicolehanjing)
* Bump k8s.io/kubernetes@v1.20.0 ([#151](https://github.com/kubernetes/cloud-provider-aws/pull/151), @nicolehanjing)
* Fix cloudbuild image name ([#163](https://github.com/kubernetes/cloud-provider-aws/pull/163), @nckturner)
* Fix cloudbuild and simplify Makefile & cloudbuild ([#162](https://github.com/kubernetes/cloud-provider-aws/pull/162), @nckturner)
* Add docs publishing script and improve documentation ([#161](https://github.com/kubernetes/cloud-provider-aws/pull/161), @nckturner)
* Fix build command ([#159](https://github.com/kubernetes/cloud-provider-aws/pull/159), @ayberk)
* Makefile target to build and push image for release ([#138](https://github.com/kubernetes/cloud-provider-aws/pull/138), @nckturner)
* add verify-codegen in CI check ([#153](https://github.com/kubernetes/cloud-provider-aws/pull/153), @nicolehanjing)
* Add cloud config for tags ([#152](https://github.com/kubernetes/cloud-provider-aws/pull/152), @nicolehanjing)
* Bump the go version to v1.15 latest ([#150](https://github.com/kubernetes/cloud-provider-aws/pull/150), @nicolehanjing)
* Update go modules to include latest k8s.io/kubernetes module on v-1.19 ([#146](https://github.com/kubernetes/cloud-provider-aws/pull/146), @nicolehanjing)
* Update Flags section of README ([#147](https://github.com/kubernetes/cloud-provider-aws/pull/147), @ayberk)
* instances: initial implementation of instancesV2 interface ([#131](https://github.com/kubernetes/cloud-provider-aws/pull/131), @nicolehanjing)
* latest manifest should point to v1.19.0-alpha.1, not v1.19.1-alpha.1 ([#140](https://github.com/kubernetes/cloud-provider-aws/pull/140), @andrewsykim)

Special notes for your reviewer:
Copy pasted from the authenticator as is.

Does this PR introduce a user-facing change?:

NONE

@k8s-ci-robot k8s-ci-robot added release-note-none Denotes a PR that doesn't merit a release note. kind/documentation Categorizes issue or PR as related to documentation. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. labels Feb 19, 2021
@k8s-ci-robot k8s-ci-robot added the size/M Denotes a PR that changes 30-99 lines, ignoring generated files. label Feb 19, 2021
@nckturner

Copy link
Copy Markdown
Contributor

/lgtm

@k8s-ci-robot k8s-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Feb 22, 2021
@k8s-ci-robot k8s-ci-robot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Feb 22, 2021
@k8s-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: ayberk, nckturner

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@k8s-ci-robot
k8s-ci-robot merged commit 0f0d7e5 into kubernetes:master Feb 22, 2021
@ayberk
ayberk deleted the changelog_script branch April 29, 2021 19:51
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. kind/documentation Categorizes issue or PR as related to documentation. lgtm "Looks good to me", indicates that a PR is ready to be merged. release-note-none Denotes a PR that doesn't merit a release note. size/M Denotes a PR that changes 30-99 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants