Skip to content

Add support for RBN-based node names - #286

Merged
k8s-ci-robot merged 1 commit into
kubernetes:masterfrom
olemarkus:rbn
Nov 30, 2021
Merged

Add support for RBN-based node names#286
k8s-ci-robot merged 1 commit into
kubernetes:masterfrom
olemarkus:rbn

Conversation

@olemarkus

@olemarkus olemarkus commented Nov 24, 2021

Copy link
Copy Markdown
Member

What type of PR is this?

Uncomment only one, leave it on its own line:

/kind feature

What this PR does / why we need it:

AWS recently launched support for instance DNS host names based on instance ID rather than ipv4.
This PR adds support for nodes with those DNS hostnames

Which issue(s) this PR fixes:

Fixes #285

Special notes for your reviewer:

Originally, node names were suffixed with the domain name. AWS CCM is to my knowledge the only CCM that do not use only the hostnames. As a consequence, all relevant k8s components need set --hostname-override as kubelet et.al cannot automatically determine the hostname. Because of this, with RBN, CCM expects only the hostname. The kubectl get nodes output then looks like this:

$ kubectl get nodes
NAME                  STATUS   ROLES                  AGE   VERSION
i-03a7726a034596f51   Ready    node                   81m   v1.21.3
i-0e0d7ce03d5827f91   Ready    control-plane,master   91m   v1.21.3

Does this PR introduce a user-facing change?:

Added support for instances with Resource Based Naming.

@k8s-ci-robot k8s-ci-robot added release-note Denotes a PR that will be considered when it comes time to generate release notes. needs-kind Indicates a PR lacks a `kind/foo` label and requires one. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. labels Nov 24, 2021
Comment thread pkg/providers/v1/aws.go Outdated
Update pkg/providers/v1/aws.go

Co-authored-by: John Gardiner Myers <jgmyers@proofpoint.com>
@johngmyers

Copy link
Copy Markdown
Member

/lgtm

@k8s-ci-robot k8s-ci-robot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Nov 28, 2021
@johngmyers

Copy link
Copy Markdown
Member

/kind feature
/assign @andrewsykim

@k8s-ci-robot k8s-ci-robot added kind/feature Categorizes issue or PR as related to a new feature. and removed needs-kind Indicates a PR lacks a `kind/foo` label and requires one. labels Nov 28, 2021

@nckturner nckturner left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

This looks good to me, as it should only affect RBN named instance, afaict.
/lgtm
/approve

@k8s-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: johngmyers, nckturner, olemarkus

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. kind/feature Categorizes issue or PR as related to a new feature. lgtm "Looks good to me", indicates that a PR is ready to be merged. release-note Denotes a PR that will be considered when it comes time to generate release notes. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

Node controller fails to get provider ID for instance with Resource Based Name

5 participants