Add environment and securityContexts - #328
Conversation
|
@jkroepke: This issue is currently awaiting triage. If cloud-provider-aws contributors determine this is a relevant issue, they will accept it by applying the The DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. |
|
Welcome @jkroepke! |
|
Hi @jkroepke. Thanks for your PR. I'm waiting for a kubernetes member to verify that this patch is reasonable to test. If it is, they should reply with Once the patch is verified, the new status will be reflected by the I understand the commands that are listed here. DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. |
|
/assign @andrewsykim |
| # env -- Pod environment variables | ||
| env: {} | ||
|
|
||
| hostNetworking: false |
There was a problem hiding this comment.
This won't work. CCM in many cases need to work before CNI and therefore must use the host's network.
There was a problem hiding this comment.
Hey @olemarkus
please correct me if I'm wrong, but I'm only declare the current default here. I will remove it from PR, since it's out of scope.
Anyway, why CCM is needed for an CNI? https://github.com/aws/amazon-vpc-cni-k8s does not depend on CCM and ignores any taints by default.
Please explain a case, I guess I miss one.
There was a problem hiding this comment.
See https://kubernetes.io/docs/concepts/architecture/cloud-controller/#route-controller
Its fine to have this configurable, although I don't think that serves much purpose. The point of using IRSA with CCM is typically to remove privileges from the control plane role, not protect the control plane role (which is usually the case otherwise). Thus that it runs hostNetworking is not a problem.
While the VPC CNI helm chart tolerates all taints, it is not given that other CNIs' install instructions do so. Which means that the CNI DS and CCM DS block each other.
There was a problem hiding this comment.
Thanks!
Its fine to have this configurable, although I don't think that serves much purpose.
It's already configurable on master, introduced in #173. The if condition leads to an undefined value by default.
8a09911 to
bd69c82
Compare
|
/ok-to-test |
|
Hi, is everything fine or some changes are still required? |
|
/lgtm |
|
/approve |
|
[APPROVALNOTIFIER] This PR is APPROVED This pull-request has been approved by: jkroepke, nckturner, olemarkus The full list of commands accepted by this bot can be found here. The pull request process is described here DetailsNeeds approval from an approver in each of these files:
Approvers can indicate their approval by writing |
What type of PR is this?
/kind cleanup
What this PR does / why we need it:
Support IRSA for CCM and securityContext for additional hardening.
Which issue(s) this PR fixes:
Fixes #327
Special notes for your reviewer:
Does this PR introduce a user-facing change?: