KEP 4680: Update README.md for DRA - #5302
Conversation
This update transfers the design details from https://docs.google.com/document/d/1CopUHEeTebGeSb3jvmHkpABACAnrDFj5OCmEqRSKCDY/edit?usp=sharing
|
Welcome @Jpsassine! |
|
Hi @Jpsassine. Thanks for your PR. I'm waiting for a kubernetes member to verify that this patch is reasonable to test. If it is, they should reply with Once the patch is verified, the new status will be reflected by the I understand the commands that are listed here. DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. |
|
/ok-to-test |
|
/cc @klueska |
|
/wg device-management |
pohly
left a comment
There was a problem hiding this comment.
No major concerns from me about the design. It's quite a bit of additional code and complexity and feels a bit redundant, but I understand that surfacing the health information in the pod status needs it.
Some nit and suggestions.
| string pool_name = 1; | ||
| // The unique name of the device within the pool. | ||
| // Required. | ||
| string device_name = 2; |
There was a problem hiding this comment.
xref: #5104 (comment)
TLDR: "device name" will become <device name>[/<share ID>] once consumable capacities allow allocating the same device multiple times. This is consistent with the proposed API here, it just needs an updated description - but not now! It depends on the order in which the implementations get merged.
cc @sunya-ch
Previously the design dug too much into the code details which is not the focus. Now it focuses on the higher level design and not the direct code implementation.
Ran ./hack/update-toc.sh
|
[APPROVALNOTIFIER] This PR is APPROVED This pull-request has been approved by: Jpsassine, mrunalp The full list of commands accepted by this bot can be found here. The pull request process is described here DetailsNeeds approval from an approver in each of these files:
Approvers can indicate their approval by writing |
|
/lgtm |
This update transfers the design details from https://docs.google.com/document/d/1CopUHEeTebGeSb3jvmHkpABACAnrDFj5OCmEqRSKCDY/edit?usp=sharing
Issue link: Add Resource Health Status to the Pod Status for Device Plugin and DRA #4680
Reference doc https://docs.google.com/document/d/1CopUHEeTebGeSb3jvmHkpABACAnrDFj5OCmEqRSKCDY/edit?usp=sharing