Skip to content

second trigger for cs-fetch-repos image initial build & push - #3528

Merged
k8s-ci-robot merged 1 commit into
kubernetes:mainfrom
Priyankasaggu11929:patch-2
Mar 17, 2022
Merged

second trigger for cs-fetch-repos image initial build & push#3528
k8s-ci-robot merged 1 commit into
kubernetes:mainfrom
Priyankasaggu11929:patch-2

Conversation

@Priyankasaggu11929

Copy link
Copy Markdown
Member

The PR is for re-triggering the currently failing cs-fetch-repos image build/push post-submit job ~ https://prow.k8s.io/view/gs/kubernetes-jenkins/logs/post-k8sio-push-image-cs-fetch-repos-canary/1504388922468208640

cc: @ameukam

@k8s-ci-robot k8s-ci-robot added cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. sig/k8s-infra Categorizes an issue or PR as relevant to SIG K8s Infra. labels Mar 17, 2022
@k8s-ci-robot
k8s-ci-robot requested review from spiffxp and thockin March 17, 2022 10:57
@ameukam

ameukam commented Mar 17, 2022

Copy link
Copy Markdown
Member

/lgtm
/approve

@k8s-ci-robot k8s-ci-robot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Mar 17, 2022
@k8s-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: ameukam, Priyankasaggu11929

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@k8s-ci-robot k8s-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Mar 17, 2022
@k8s-ci-robot
k8s-ci-robot merged commit c16ef42 into kubernetes:main Mar 17, 2022
@k8s-ci-robot k8s-ci-robot added this to the v1.24 milestone Mar 17, 2022
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. sig/k8s-infra Categorizes an issue or PR as relevant to SIG K8s Infra. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants