Sync community-values.md from k/community - #34311
Conversation
✅ Pull request preview available for checkingBuilt without sensitive environment variables
To edit notification comments on pull requests, go to your Netlify site settings. |
zacharysarah
left a comment
There was a problem hiding this comment.
@Sea-n Thanks for updating this. ✨
The preview now centers body text, but I don't see any formatting or CSS changes. It's not blocking, just noticeable. 🤷🏻
Otherwise LGTM
|
I'm willing to give an LGTM but would prefer a co-chair perform the approval. 👋🏻 /lgtm |
|
LGTM label has been added. DetailsGit tree hash: 365e50849fdf7b8ad7299bbdb47a30bb1a9c17ac |
|
/assign @natalisucks @reylejano @divya-mohan0209 |
|
Oh, thanks @zacharysarah for pointing the style mismatch out! I've fixed that, waiting for others to review. |
|
/lgtm |
|
LGTM label has been added. DetailsGit tree hash: 6a553625c4d02daf5950689086544cd1d043fe82 |
divya-mohan0209
left a comment
There was a problem hiding this comment.
Thank you for bringing this to our attention @zacharysarah and thank you so much for fixing this @Sea-n !
/approve
|
[APPROVALNOTIFIER] This PR is APPROVED This pull-request has been approved by: divya-mohan0209, zacharysarah The full list of commands accepted by this bot can be found here. The pull request process is described here DetailsNeeds approval from an approver in each of these files:
Approvers can indicate their approval by writing |
Apply changes from January 2021.
Add similar text adopted from /content/en/comm/coc.md to link to the canonical version.
Netlify preview: https://deploy-preview-34311--kubernetes-io-main-staging.netlify.app/community/values