Skip to content

Bug 1891552: Handle missing labels as empty - #6940

Merged
openshift-merge-robot merged 1 commit into
openshift:masterfrom
appuio:fix-alert-silence-matcher
Oct 21, 2020
Merged

Bug 1891552: Handle missing labels as empty#6940
openshift-merge-robot merged 1 commit into
openshift:masterfrom
appuio:fix-alert-silence-matcher

Conversation

@srueg

@srueg srueg commented Oct 15, 2020

Copy link
Copy Markdown
Contributor

To implement the same logic as in AlertManager which handles missing
labels as the empty label [1].

[1] https://github.com/prometheus/alertmanager/blob/a8ed0d58509d0251adcb159552b9ab098e206f1f/types/match.go#L72

Solves https://access.redhat.com/support/cases/#/case/02776669

@openshift-ci-robot openshift-ci-robot added the component/core Related to console core functionality label Oct 15, 2020
@openshift-ci-robot

Copy link
Copy Markdown
Contributor

Hi @srueg. Thanks for your PR.

I'm waiting for a openshift member to verify that this patch is reasonable to test. If it is, they should reply with /ok-to-test on its own line. Until that is done, I will not automatically test new commits in this PR, but the usual testing commands by org members will still work. Regular contributors should join the org to skip this step.

Once the patch is verified, the new status will be reflected by the ok-to-test label.

I understand the commands that are listed here.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@openshift-ci-robot openshift-ci-robot added the needs-ok-to-test Indicates a PR that requires an org member to verify it is safe to test. label Oct 15, 2020
@srueg
srueg force-pushed the fix-alert-silence-matcher branch from 1b5779c to d08f546 Compare October 15, 2020 12:21
@spadgett

Copy link
Copy Markdown
Member

/assign @kyoto
/ok-to-test

@openshift-ci-robot openshift-ci-robot added ok-to-test Indicates a non-member PR verified by an org member that is safe to test. and removed needs-ok-to-test Indicates a PR that requires an org member to verify it is safe to test. labels Oct 15, 2020
To implement the same logic as in AlertManager which handles missing
labels as the empty label [1].

[1] https://github.com/prometheus/alertmanager/blob/a8ed0d58509d0251adcb159552b9ab098e206f1f/types/match.go#L72

Signed-off-by: Simon Rüegg <simon@rueggs.ch>
@srueg
srueg force-pushed the fix-alert-silence-matcher branch from d08f546 to 7548e7f Compare October 15, 2020 15:00
@kyoto

kyoto commented Oct 19, 2020

Copy link
Copy Markdown
Member

Thank you for the PR @srueg!

@simonpasquier Helped me confirm the issue and the code change looks good to me.

I'd like to do some more testing first, then hopefully we can merge.

@kyoto

kyoto commented Oct 21, 2020

Copy link
Copy Markdown
Member

/lgtm

Tested locally and compared with Alertmanager behavior. Thanks again @srueg!

@openshift-ci-robot openshift-ci-robot added the lgtm Indicates that a PR is ready to be merged. label Oct 21, 2020
@openshift-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: kyoto, srueg

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci-robot openshift-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Oct 21, 2020
@openshift-merge-robot
openshift-merge-robot merged commit b52717d into openshift:master Oct 21, 2020
@srueg
srueg deleted the fix-alert-silence-matcher branch October 21, 2020 06:05
@srueg

srueg commented Oct 21, 2020

Copy link
Copy Markdown
Contributor Author

Cool, thanks for the quick handling!
What's the process for releasing this? Can we get this back-ported to v4.5 ?

@spadgett spadgett added this to the v4.7 milestone Oct 21, 2020
@germanparente

Copy link
Copy Markdown
Contributor

I have just reported:
https://bugzilla.redhat.com/show_bug.cgi?id=1891552
Bug 1891552 - Handle missing labels as empty.

to include this fix downstream.

@kyoto

kyoto commented Oct 27, 2020

Copy link
Copy Markdown
Member

/cherrypick release-4.6

@openshift-cherrypick-robot

Copy link
Copy Markdown

@kyoto: new pull request created: #7002

Details

In response to this:

/cherrypick release-4.6

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@openshift-ci-robot openshift-ci-robot changed the title Handle missing labels as empty [release-4.6] Bug 1891552: Handle missing labels as empty Oct 27, 2020
@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@srueg: All pull requests linked via external trackers have merged:

Bugzilla bug 1891552 has been moved to the MODIFIED state.

Details

In response to this:

[release-4.6] Bug 1891552: Handle missing labels as empty

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@kyoto

kyoto commented Oct 27, 2020

Copy link
Copy Markdown
Member

/retitle Bug 1891552: Handle missing labels as empty

@openshift-ci-robot openshift-ci-robot changed the title [release-4.6] Bug 1891552: Handle missing labels as empty Bug 1891552: Handle missing labels as empty Oct 27, 2020
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. component/core Related to console core functionality lgtm Indicates that a PR is ready to be merged. ok-to-test Indicates a non-member PR verified by an org member that is safe to test.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

7 participants