Skip to content

Add support for Jekyll 1.0.0 - #7

Closed
cgarvis wants to merge 6 commits into
mattmanning:masterfrom
cgarvis:patch-1
Closed

Add support for Jekyll 1.0.0#7
cgarvis wants to merge 6 commits into
mattmanning:masterfrom
cgarvis:patch-1

Conversation

@cgarvis

@cgarvis cgarvis commented Apr 30, 2013

Copy link
Copy Markdown

No description provided.

@cgarvis

cgarvis commented May 1, 2013

Copy link
Copy Markdown
Author

@mattmanning I'm having some issues getting Jekyll to build on heroku. I'm getting the following error message

ERROR: YOUR SITE COULD NOT BE BUILT:
              ------------------------------------
              Post 0000-00-00-welcome-to-jekyll.markdown.erb does not have a valid date.

That file is a test file. It looks like Jekyll is building on top of itself. Does heroku install the gems in the same path as the app?

@brandonweiss

Copy link
Copy Markdown

I had this same problem. You need to exclude the vendor directory in your configuration. Like so: exclude: ["vendor"].

@cgarvis

cgarvis commented May 10, 2013

Copy link
Copy Markdown
Author

I've since moved to doing the jekyll build using the asset compiler. If I where to finish this out, I would create a config for this that could then be merged with project's jekyll config. I believe Jekyll 1.0 now supports this.

@brandonweiss

Copy link
Copy Markdown

Oh, interesting, so you've started doing all the building and whatnot via rake assets:precompile?

@oddsve

oddsve commented May 24, 2013

Copy link
Copy Markdown

👍

1 similar comment
@parkr

parkr commented May 29, 2013

Copy link
Copy Markdown

👍

@ryanmasondavies

Copy link
Copy Markdown

Would be great to see this merged 👍

@cgarvis

cgarvis commented Aug 20, 2013

Copy link
Copy Markdown
Author

You can see how I implemented it on my blog's rakefile:

https://github.com/cgarvis/chrisgarvis.com/blob/master/Rakefile#L5

@cgarvis cgarvis closed this Aug 20, 2013
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

5 participants