Skip to content

bump /usr partition to 6800 MB for anssi_bp28 tests in rhel8 - #15025

Merged
ggbecker merged 1 commit into
ComplianceAsCode:masterfrom
Arden97:adjust_ks_part
Aug 21, 2026
Merged

bump /usr partition to 6800 MB for anssi_bp28 tests in rhel8#15025
ggbecker merged 1 commit into
ComplianceAsCode:masterfrom
Arden97:adjust_ks_part

Conversation

@Arden97

@Arden97 Arden97 commented Aug 20, 2026

Copy link
Copy Markdown
Member

Description:

  • Bump /usr partition to 6800 MB for some of the /hardening/anaconda/with-gui/anssi_bp28 tests on RHEL 8

Rationale:

The following tests failed in recent stabilization run failed because of disk space error.

  • /hardening/anaconda/with-gui/anssi_bp28_intermediary - at least 33MB more space required
  • /hardening/anaconda/with-gui/anssi_bp28_enhanced - at least 33MB more space required
  • /hardening/anaconda/with-gui/anssi_bp28_high - at least 28MB more space required

@openshift-ci

openshift-ci Bot commented Aug 20, 2026

Copy link
Copy Markdown

Skipping CI for Draft Pull Request.
If you want CI signal for your change, please convert it to an actual PR.
You can still manually trigger a test run with /test all

@openshift-ci openshift-ci Bot added the do-not-merge/work-in-progress Used by openshift-ci bot. label Aug 20, 2026
@Arden97 Arden97 modified the milestones: 0.1.83, 0.1.82 Aug 20, 2026
@Arden97
Arden97 marked this pull request as ready for review August 20, 2026 10:51
@openshift-ci openshift-ci Bot removed the do-not-merge/work-in-progress Used by openshift-ci bot. label Aug 20, 2026
@Arden97

Arden97 commented Aug 20, 2026

Copy link
Copy Markdown
Member Author

Custom productization passed for mentioned tests

@Arden97
Arden97 requested a review from Mab879 August 20, 2026 11:07
@openshift-ci

openshift-ci Bot commented Aug 20, 2026

Copy link
Copy Markdown

@Arden97: The following tests failed, say /retest to rerun all failed tests or /retest-required to rerun all mandatory failed tests:

Test name Commit Details Required Rerun command
ci/prow/e2e-aws-openshift-platform-compliance 114f737 link true /test e2e-aws-openshift-platform-compliance
ci/prow/e2e-aws-openshift-node-compliance 114f737 link false /test e2e-aws-openshift-node-compliance

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

@comps

comps commented Aug 21, 2026

Copy link
Copy Markdown
Collaborator

As a side note to @Mab879 and others who struggled to fit within 20 GB - it's no longer an issue for Contest ever since we fixed RHSecurityCompliance/contest#559 - the disks are now 100 GB of virtual space (can easily be more if needed) and we might want to restructure perhaps most upstream kickstarts to have a bigger reserve.

(I don't think we promise anywhere that they should fit within 20 GB, it was just a limitation of Contest.)

@ggbecker ggbecker left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Looks good to me:

The difference is 133 MB between the two versions. And still fits within the 20GB disk requirement.

@ggbecker ggbecker self-assigned this Aug 21, 2026
@ggbecker

Copy link
Copy Markdown
Member

As a side note to @Mab879 and others who struggled to fit within 20 GB - it's no longer an issue for Contest ever since we fixed RHSecurityCompliance/contest#559 - the disks are now 100 GB of virtual space (can easily be more if needed) and we might want to restructure perhaps most upstream kickstarts to have a bigger reserve.

(I don't think we promise anywhere that they should fit within 20 GB, it was just a limitation of Contest.)

Let's create an issue with this information then. It's true that we ship these kickstart files so customers can use as an example to create their own kickstart files, I guess limiting the disk size was also something we would see as minimum recommended sizes so RHEL would be able to install for example... but I guess the use cases have mixed up here as we also use these kickstarts for testing.

@ggbecker

Copy link
Copy Markdown
Member

As a side note to @Mab879 and others who struggled to fit within 20 GB - it's no longer an issue for Contest ever since we fixed RHSecurityCompliance/contest#559 - the disks are now 100 GB of virtual space (can easily be more if needed) and we might want to restructure perhaps most upstream kickstarts to have a bigger reserve.
(I don't think we promise anywhere that they should fit within 20 GB, it was just a limitation of Contest.)

Let's create an issue with this information then. It's true that we ship these kickstart files so customers can use as an example to create their own kickstart files, I guess limiting the disk size was also something we would see as minimum recommended sizes so RHEL would be able to install for example... but I guess the use cases have mixed up here as we also use these kickstarts for testing.

But we can follow up on this discussion and for now merge this one to unblock any testing required with these new sizes.

@ggbecker
ggbecker merged commit 6372619 into ComplianceAsCode:master Aug 21, 2026
72 of 75 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants