Skip to content

refactor(controls)!: remove set and get prefixes - #514

Merged
fraxken merged 1 commit into
mainfrom
refactor-controls-naming
Aug 23, 2026
Merged

refactor(controls)!: remove set and get prefixes#514
fraxken merged 1 commit into
mainfrom
refactor-controls-naming

Conversation

@fraxken

@fraxken fraxken commented Aug 23, 2026

Copy link
Copy Markdown
Member

No description provided.

@changeset-bot

changeset-bot Bot commented Aug 23, 2026

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: d4d0142

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 6 packages
Name Type
@jolly-pixel/controls Major
@jolly-pixel/engine Major
@jolly-pixel/runtime Patch
@jolly-pixel/voxel.renderer Patch
@jolly-pixel/editor.voxel-map Patch
@jolly-pixel/editor.voxel-model Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@fraxken
fraxken merged commit cd200eb into main Aug 23, 2026
7 checks passed
@fraxken
fraxken deleted the refactor-controls-naming branch August 23, 2026 18:41
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant