Skip to content

Add bluetooth-developers team to CODEOWNERS. - #20

Merged
Nga Nguyen (silabs-NgaNg) merged 2 commits into
mainfrom
feature/add-bluetooth-developers-codeowners
Sep 4, 2026
Merged

Add bluetooth-developers team to CODEOWNERS.#20
Nga Nguyen (silabs-NgaNg) merged 2 commits into
mainfrom
feature/add-bluetooth-developers-codeowners

Conversation

@silabs-NgaNg

@silabs-NgaNg Nga Nguyen (silabs-NgaNg) commented Sep 4, 2026

Copy link
Copy Markdown
Contributor

Share default review ownership with techpubs so Bluetooth developers are requested on PRs.

Description

Add bluetooth dev teams to the codeowners file so they can review the PR.

Related Issue

N/A

Changes Made

Checklist

Screenshots (if applicable)

N/A

Additional Notes

N/A

Share default review ownership with techpubs so Bluetooth developers are requested on PRs.
@silabs-NgaNg
Nga Nguyen (silabs-NgaNg) merged commit 89ff484 into main Sep 4, 2026
2 checks passed
@silabs-NgaNg
Nga Nguyen (silabs-NgaNg) deleted the feature/add-bluetooth-developers-codeowners branch September 4, 2026 14:39
@github-actions github-actions Bot locked and limited conversation to collaborators Sep 4, 2026
Sign up for free to subscribe to this conversation on GitHub. Already have an account? Sign in.

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant