Skip to content

[release/v1.65] feat: added support for NVMe disk controller for Azure VMs - #2076

Open
kubermatic-bot wants to merge 2 commits into
kubermatic:release/v1.65from
kubermatic-bot:cherry-pick-2071-to-release/v1.65
Open

[release/v1.65] feat: added support for NVMe disk controller for Azure VMs#2076
kubermatic-bot wants to merge 2 commits into
kubermatic:release/v1.65from
kubermatic-bot:cherry-pick-2071-to-release/v1.65

Conversation

@kubermatic-bot

Copy link
Copy Markdown
Contributor

This is an automated cherry-pick of #2071

/assign dharapvj

Azure VMs with series v6+ can be created via machine-controller

dharapvj and others added 2 commits September 3, 2026 19:58
Co-authored-by: Artiom Diomin <kron82@gmail.com>
Signed-off-by: Vijay Dharap <dharapvj@gmail.com>
@kubermatic-bot kubermatic-bot added release-note Denotes a PR that will be considered when it comes time to generate release notes. dco-signoff: yes Denotes that all commits in the pull request have the valid DCO signoff message. labels Sep 3, 2026
@kubermatic-bot

Copy link
Copy Markdown
Contributor Author

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by:
Once this PR has been reviewed and has the lgtm label, please assign adoi for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@kubermatic-bot kubermatic-bot added the size/L Denotes a PR that changes 100-499 lines, ignoring generated files. label Sep 3, 2026
@kubermatic-bot

Copy link
Copy Markdown
Contributor Author

@kubermatic-bot: The following test failed, say /retest to rerun all failed tests or /retest-required to rerun all mandatory failed tests:

Test name Commit Details Required Rerun command
pull-machine-controller-e2e-azure 53bdb2f link true /test pull-machine-controller-e2e-azure

Full PR test history

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

dco-signoff: yes Denotes that all commits in the pull request have the valid DCO signoff message. release-note Denotes a PR that will be considered when it comes time to generate release notes. size/L Denotes a PR that changes 100-499 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants