feat(rbac): update rbac plugin - #4505
Conversation
3bd9b8a to
c6593b0
Compare
|
The container image build workflow finished with status: |
|
/test e2e-ocp-helm |
PatAKnight
left a comment
There was a problem hiding this comment.
Tested and it worked for me.
Might want to go through and go ahead and bump all of the rbac plugins @backstage-community/plugin-rbac-* while you are at it.
a5982d3 to
cc2d84b
Compare
|
/retest |
|
The container image build workflow finished with status: |
7f4d2f1 to
ef862aa
Compare
|
The container image build workflow finished with status: |
|
The container image build workflow finished with status: |
Done. |
fd99504 to
4e8dba8
Compare
|
/retest |
4e8dba8 to
6666c18
Compare
Signed-off-by: Oleksandr Andriienko <oandriie@redhat.com>
6666c18 to
b95df06
Compare
|
The container image build workflow finished with status: |
|
c6a928c
into
redhat-developer:main



Description
Update rbac plugin
Which issue(s) does this PR fix
PR acceptance criteria
Please make sure that the following steps are complete:
How to test changes / Special notes to the reviewer