Skip to content

Input components - #1668

Merged
mcrascal merged 11 commits into
nextfrom
input-components-redux
Mar 6, 2024
Merged

Input components #1668
mcrascal merged 11 commits into
nextfrom
input-components-redux

Conversation

@mcrascal

@mcrascal mcrascal commented Mar 1, 2024

Copy link
Copy Markdown
Member

Breaks input components out of 1569

@changeset-bot

changeset-bot Bot commented Mar 1, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: a51e5a3

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 4 packages
Name Type
@evidence-dev/core-components Patch
@evidence-dev/evidence Patch
@evidence-dev/components Patch
evidence-test-environment Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@vercel

vercel Bot commented Mar 1, 2024

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for Git ↗︎

Name Status Preview Comments Updated (UTC)
docs ✅ Ready (Inspect) Visit Preview 💬 Add feedback Mar 6, 2024 1:50am

@netlify

netlify Bot commented Mar 1, 2024

Copy link
Copy Markdown

Deploy Preview for evidence-development-workspace ready!

Name Link
🔨 Latest commit a51e5a3
🔍 Latest deploy log https://app.netlify.com/sites/evidence-development-workspace/deploys/65e7cbe790d9360008fcdd7f
😎 Deploy Preview https://deploy-preview-1668--evidence-development-workspace.netlify.app
📱 Preview on mobile
Toggle QR Code...

QR Code

Use your smartphone camera to open QR code link.

To edit notification comments on pull requests, go to your Netlify site configuration.

@csjh

csjh commented Mar 5, 2024

Copy link
Copy Markdown
Contributor

I'm just noticing/remembering I never concluded the dropdown backwards compatibility fix

I'm leaning on the side of an includeLabels prop since it seems cleaner for everyone involved

@archiewood

archiewood commented Mar 5, 2024

Copy link
Copy Markdown
Contributor

how would that work?

<Dropdown includeLabels=true/> allows

${inputs.my_input.value} and ${inputs.my_input.label}?

but
<Dropdown/>

only allows

${inputs.my_input}

@csjh

csjh commented Mar 5, 2024

Copy link
Copy Markdown
Contributor

yup

@archiewood

Copy link
Copy Markdown
Contributor

Sounds like a good solution to me.

What do you think about warning a breaking change in the future?

${inputs.my_input} will be deprecated in later versions of evidence.

Use ${inputs.my_input.value} in combination with includeLabels=true

@csjh

csjh commented Mar 5, 2024

Copy link
Copy Markdown
Contributor

Yeah makes sense to me

@mcrascal

mcrascal commented Mar 6, 2024

Copy link
Copy Markdown
Member Author

@archiewood I'm inclined to just make the breaking change and announce it in the changelog

@mcrascal
mcrascal merged commit 72f9833 into next Mar 6, 2024
@mcrascal
mcrascal deleted the input-components-redux branch March 6, 2024 02:11
@archiewood

Copy link
Copy Markdown
Contributor

That's fine. We should definitely add a major bump in that case

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants