Skip to content

Remove headless UI - #1700

Merged
mcrascal merged 4 commits into
nextfrom
remove-headless-ui
Mar 7, 2024
Merged

Remove headless UI#1700
mcrascal merged 4 commits into
nextfrom
remove-headless-ui

Conversation

@mcrascal

@mcrascal mcrascal commented Mar 7, 2024

Copy link
Copy Markdown
Member

Removes headless UI dependency, moves the kebab over to shadcn, hides the kebab entirely in production when never show queries is true.

@changeset-bot

changeset-bot Bot commented Mar 7, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 55341e5

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 4 packages
Name Type
@evidence-dev/core-components Patch
@evidence-dev/evidence Patch
@evidence-dev/components Patch
evidence-test-environment Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@mcrascal
mcrascal requested a review from ItsMeBrianD March 7, 2024 16:36
@vercel

vercel Bot commented Mar 7, 2024

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for Git ↗︎

Name Status Preview Comments Updated (UTC)
docs ✅ Ready (Inspect) Visit Preview 💬 Add feedback Mar 7, 2024 4:45pm

@netlify

netlify Bot commented Mar 7, 2024

Copy link
Copy Markdown

Deploy Preview for evidence-development-workspace ready!

Name Link
🔨 Latest commit 55341e5
🔍 Latest deploy log https://app.netlify.com/sites/evidence-development-workspace/deploys/65e9ef1a6164c70008915ea1
😎 Deploy Preview https://deploy-preview-1700--evidence-development-workspace.netlify.app
📱 Preview on mobile
Toggle QR Code...

QR Code

Use your smartphone camera to open QR code link.

To edit notification comments on pull requests, go to your Netlify site configuration.

@archiewood

Copy link
Copy Markdown
Contributor

@mcrascal will this prevent people printing to PDF when never show queries is true?

@mcrascal

mcrascal commented Mar 7, 2024

Copy link
Copy Markdown
Member Author

It will remove the button that says "print pdf" but it will have no effect on people's ability to print the page. For our docs at least, I don't want to have a dropdown menu with one sort of useless item in it.

@mcrascal
mcrascal merged commit 2d0c00b into next Mar 7, 2024
@mcrascal
mcrascal deleted the remove-headless-ui branch March 7, 2024 18:50
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants