Skip to content

Roll back proxy server - #1714

Merged
ItsMeBrianD merged 2 commits into
nextfrom
rollback-proxy
Mar 8, 2024
Merged

Roll back proxy server#1714
ItsMeBrianD merged 2 commits into
nextfrom
rollback-proxy

Conversation

@hughess

@hughess hughess commented Mar 8, 2024

Copy link
Copy Markdown
Member

Description

Rolls back changes from proxy server loading feedback until we can fix the behaviour across dev environments. Replaces with a console message.

Checklist

@changeset-bot

changeset-bot Bot commented Mar 8, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 08cbf7f

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 11 packages
Name Type
@evidence-dev/component-utilities Patch
@evidence-dev/core-components Patch
@evidence-dev/evidence Patch
@evidence-dev/plugin-connector Patch
@evidence-dev/preprocess Patch
@evidence-dev/query-store Patch
@evidence-dev/universal-sql Patch
evidence-docs Patch
@evidence-dev/components Patch
evidence-test-environment Patch
ui-tests Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@vercel

vercel Bot commented Mar 8, 2024

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for Git ↗︎

Name Status Preview Comments Updated (UTC)
docs ✅ Ready (Inspect) Visit Preview 💬 Add feedback Mar 8, 2024 4:36pm

@netlify

netlify Bot commented Mar 8, 2024

Copy link
Copy Markdown

Deploy Preview for evidence-development-workspace ready!

Name Link
🔨 Latest commit 08cbf7f
🔍 Latest deploy log https://app.netlify.com/sites/evidence-development-workspace/deploys/65eb3e659331f60008a21a0b
😎 Deploy Preview https://deploy-preview-1714--evidence-development-workspace.netlify.app
📱 Preview on mobile
Toggle QR Code...

QR Code

Use your smartphone camera to open QR code link.

To edit notification comments on pull requests, go to your Netlify site configuration.

@ItsMeBrianD
ItsMeBrianD merged commit 580256d into next Mar 8, 2024
@ItsMeBrianD
ItsMeBrianD deleted the rollback-proxy branch March 8, 2024 17:21
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants