Skip to content

Patch Release - #1712

Merged
mcrascal merged 8 commits into
mainfrom
next
Mar 8, 2024
Merged

Patch Release#1712
mcrascal merged 8 commits into
mainfrom
next

Conversation

@mcrascal

@mcrascal mcrascal commented Mar 8, 2024

Copy link
Copy Markdown
Member

No description provided.

@changeset-bot

changeset-bot Bot commented Mar 8, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 542f4e4

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 11 packages
Name Type
@evidence-dev/component-utilities Patch
@evidence-dev/core-components Patch
@evidence-dev/evidence Patch
@evidence-dev/plugin-connector Patch
@evidence-dev/preprocess Patch
@evidence-dev/query-store Patch
@evidence-dev/universal-sql Patch
evidence-docs Patch
@evidence-dev/components Patch
evidence-test-environment Patch
ui-tests Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@vercel

vercel Bot commented Mar 8, 2024

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for Git ↗︎

Name Status Preview Comments Updated (UTC)
docs ✅ Ready (Inspect) Visit Preview 💬 Add feedback Mar 8, 2024 7:18pm

@mcrascal
mcrascal requested review from ItsMeBrianD and hughess March 8, 2024 13:50
@netlify

netlify Bot commented Mar 8, 2024

Copy link
Copy Markdown

Deploy Preview for evidence-development-workspace ready!

Name Link
🔨 Latest commit 542f4e4
🔍 Latest deploy log https://app.netlify.com/sites/evidence-development-workspace/deploys/65eb6485a79f940008e10ff0
😎 Deploy Preview https://deploy-preview-1712--evidence-development-workspace.netlify.app
📱 Preview on mobile
Toggle QR Code...

QR Code

Use your smartphone camera to open QR code link.

To edit notification comments on pull requests, go to your Netlify site configuration.

@mcrascal
mcrascal merged commit 91d7e98 into main Mar 8, 2024
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants